Skip to content

Sync with react.dev @ 383a1e92 - #731

Merged
anilcanboga merged 17 commits into
mainfrom
sync-383a1e92
Oct 8, 2026
Merged

anilcanboga merged 17 commits into
mainfrom
sync-383a1e92

Conversation

@react-translations-bot

@react-translations-bot react-translations-bot commented Aug 17, 2026 •

Copy link
Copy Markdown
Collaborator

This PR was automatically generated.

Merge changes from react.dev at 383a1e9

The following files have conflicts and may need new translations:

Please fix the conflicts by pushing new commits to this pull request, either by editing the files directly on GitHub or by checking out this branch.

DO NOT SQUASH MERGE THIS PULL REQUEST!

Doing so will "erase" the commits from main and cause them to show up as conflicts the next time we merge.

@vercel

vercel Bot commented Aug 17, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
tr-legacy-reactjs-org Ignored Ignored Preview Oct 8, 2026 9:54pm UTC

Request Review

@github-actions

github-actions Bot commented Oct 8, 2026

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@anilcanboga
anilcanboga merged commit 9488551 into main Oct 8, 2026
10 checks passed

This branch was successfully deployed

1 active deployment
Preview – tr-react-dev — 2d93cea9 Deployed Oct 8, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants